Sitelet https://github.com/alauda/ui/pull/374
Skip to content

fix: backspace of multi-select, bump eslint-mdx - #374

Merged
JounQin merged 2 commits into
masterfrom
fix/multi-select-backspace
Aug 30, 2022
Merged

JounQin merged 2 commits into
masterfrom
fix/multi-select-backspace

Conversation

@EdisonSu768

@EdisonSu768 EdisonSu768 commented Aug 30, 2022 •

Copy link
Copy Markdown
Member
  • upgrade eslint-mdx and plugin to 2.0.2
  • fix backspace can delete disabled selected option in multi select

@changeset-bot

changeset-bot Bot commented Aug 30, 2022 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 809d1c1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@EdisonSu768
EdisonSu768 force-pushed the fix/multi-select-backspace branch from d5a26a1 to 5a0e9e2 Compare August 30, 2022 06:04
@EdisonSu768
EdisonSu768 force-pushed the fix/multi-select-backspace branch from 5a0e9e2 to a5b03e5 Compare August 30, 2022 08:04
@EdisonSu768
EdisonSu768 marked this pull request as ready for review August 30, 2022 08:06
@JounQin JounQin changed the title Fix/multi select backspace fix: multi-select` Aug 30, 2022
@JounQin JounQin changed the title fix: multi-select` fix: backspace of multi-select Aug 30, 2022
@EdisonSu768 EdisonSu768 added the bug Something isn't working label Aug 30, 2022
@JounQin JounQin added the dependencies Pull requests that update a dependency file label Aug 30, 2022
@JounQin JounQin changed the title fix: backspace of multi-select fix: backspace of multi-select, bump eslint-mdx Aug 30, 2022
@JounQin
JounQin merged commit a7ec1b2 into master Aug 30, 2022
@JounQin
JounQin deleted the fix/multi-select-backspace branch August 30, 2022 10:14
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

bug Something isn't working dependencies Pull requests that update a dependency file

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants