Sitelet https://web.archive.org/web/20210505031017/https://github.com/pyenv/pyenv/pull/1548
Skip to content
New issue

Have a question about this project? Sign up for a free GitHub account to open an issue and contact its maintainers and the community.

By clicking “Sign up for GitHub”, you agree to our terms of service and privacy statement. We’ll occasionally send you account related emails.

Already on GitHub? Sign in to your account

Docker config for testing python-build #1548

Merged
merged 4 commits into from Dec 11, 2020
Merged

Docker config for testing python-build #1548

merged 4 commits into from Dec 11, 2020

Conversation

@clbarnes
Copy link
Contributor

@clbarnes clbarnes commented Mar 5, 2020

An alternative to #1547 (see there for more discussion).

Pick this one if you want a faster spin up -> build python -> destroy environment cycle. Pick vagrant if you want to be able to make changes to your build script, have them persisted outside of the container, and re-test in the same container.

@clbarnes clbarnes mentioned this pull request Mar 5, 2020
4 of 5 tasks complete
@clbarnes clbarnes force-pushed the clbarnes:docker branch from 0ee5b88 to 6233795 Mar 5, 2020
@clbarnes
Copy link
Contributor Author

@clbarnes clbarnes commented Mar 5, 2020

Possible extension

Have travis check whether a commit has more versions than master, and if so, try to build every one of them. It may take a while but it will prove that new additions work.

dockerignore file, specify ephemeral nature of containers.
@clbarnes clbarnes mentioned this pull request Mar 5, 2020
@clbarnes
Copy link
Contributor Author

@clbarnes clbarnes commented Mar 6, 2020 •

On the testing point, you can test all new or changed versions compared to master with

set -e
set -x

docker build -t test-container .

git diff --name-only master \
  | grep '^plugins/python-build/share/python-build/' \
  | awk -F '/' '{print $NF}' \
  | xargs -I _ docker run test-container pyenv install _
@clbarnes
Copy link
Contributor Author

@clbarnes clbarnes commented Oct 8, 2020

Would be great to get a review on this! @yyuu ? I found it invaluable for testing new miniconda builds when I added a bunch a while back. The above bash script could potentially go it into CI for new build scripts.

@clbarnes
Copy link
Contributor Author

@clbarnes clbarnes commented Dec 10, 2020

I'd love to get this reviewed - @joshfriend @chrahunt , maybe? Not really sure how to move this forward.

When I was working on bulk-contributing new versions previously, nobody was able to answer how to test that they work "en masse". I still don't know how people are testing newly submitted versions. This represents a drastic improvement whatever that testing workflow is, and doesn't seem to add much maintenance overhead.

@joshfriend joshfriend merged commit 23677c5 into pyenv:master Dec 11, 2020
1 check passed
1 check passed
continuous-integration/travis-ci/pr The Travis CI build passed
Details
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment
Labels
None yet
Projects
None yet
Linked issues

Successfully merging this pull request may close these issues.

None yet

2 participants