Sitelet https://github.com/alauda/ui/pull/352
Skip to content

fix: autocomplete disabled - #352

Merged
JounQin merged 2 commits into
alauda:masterfrom
EdisonSu768:fix/autocomplete-disabled
May 26, 2022
Merged

JounQin merged 2 commits into
alauda:masterfrom
EdisonSu768:fix/autocomplete-disabled

Conversation

@EdisonSu768

Copy link
Copy Markdown
Member

autocomplete disabled not working, when adding this attr.

@changeset-bot

changeset-bot Bot commented May 25, 2022 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4e24b38

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codesandbox-ci

codesandbox-ci Bot commented May 25, 2022 •

Copy link
Copy Markdown

This pull request is automatically built and testable in CodeSandbox.

To see build info of the built libraries, click here or the icon next to each commit SHA.

@EdisonSu768
EdisonSu768 force-pushed the fix/autocomplete-disabled branch from 7f7c662 to 6e4b661 Compare May 26, 2022 03:43
Comment thread src/autocomplete/autocomplete.directive.ts Outdated
Comment thread src/autocomplete/autocomplete.directive.ts Outdated
@JounQin

JounQin commented May 26, 2022

Copy link
Copy Markdown
Member

@EdisonSu768 添加一下 changeset

@EdisonSu768

Copy link
Copy Markdown
Member Author

@EdisonSu768 添加一下 changeset

I will

@EdisonSu768
EdisonSu768 force-pushed the fix/autocomplete-disabled branch from 6e4b661 to d817150 Compare May 26, 2022 04:16
JounQin
JounQin previously approved these changes May 26, 2022
@JounQin
JounQin merged commit eb0774a into alauda:master May 26, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants