Sitelet https://github.com/alauda/ui/pull/299
Skip to content

fix: system scrollbar color - #299

Merged
fengtianze merged 2 commits into
masterfrom
fix/system-scroll-bar-color
Mar 3, 2022
Merged

fengtianze merged 2 commits into
masterfrom
fix/system-scroll-bar-color

Conversation

@fengtianze

Copy link
Copy Markdown
Collaborator

No description provided.

@changeset-bot

changeset-bot Bot commented Mar 3, 2022 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d49f0ff

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codesandbox-ci

codesandbox-ci Bot commented Mar 3, 2022

Copy link
Copy Markdown

This pull request is automatically built and testable in CodeSandbox.

To see build info of the built libraries, click here or the icon next to each commit SHA.

@fengtianze
fengtianze merged commit d3f5d22 into master Mar 3, 2022
@fengtianze
fengtianze deleted the fix/system-scroll-bar-color branch March 3, 2022 10:03
@github-actions github-actions Bot mentioned this pull request Mar 3, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant