Sitelet https://github.com/alauda/ui/pull/271
Skip to content

feat: solid icon background - #271

Merged
fengtianze merged 2 commits into
masterfrom
feat/dark-mode-optimize
Jan 25, 2022
Merged

fengtianze merged 2 commits into
masterfrom
feat/dark-mode-optimize

Conversation

@fengtianze

Copy link
Copy Markdown
Collaborator

No description provided.

@changeset-bot

changeset-bot Bot commented Jan 25, 2022 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dd5867d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codesandbox-ci

codesandbox-ci Bot commented Jan 25, 2022 •

Copy link
Copy Markdown

This pull request is automatically built and testable in CodeSandbox.

To see build info of the built libraries, click here or the icon next to each commit SHA.

@fengtianze
fengtianze force-pushed the feat/dark-mode-optimize branch from f7ab750 to 9494aed Compare January 25, 2022 09:31
@fengtianze
fengtianze merged commit b0c1e7c into master Jan 25, 2022
@fengtianze
fengtianze deleted the feat/dark-mode-optimize branch January 25, 2022 09:41
@github-actions github-actions Bot mentioned this pull request Jan 25, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant